Skip to content

Update owner from andreagit97 to andreaterzolo - #86

Merged
poiana merged 1 commit into
mainfrom
andreaterzolo-patch-1
Sep 6, 2026
Merged

Update owner from andreagit97 to andreaterzolo#86
poiana merged 1 commit into
mainfrom
andreaterzolo-patch-1

Conversation

@andreaterzolo

Copy link
Copy Markdown
Contributor

What type of PR is this?

Uncomment one (or more) /kind <> lines:

/kind bug

/kind cleanup

/kind design

/kind documentation

/kind failing-test

/kind feature

Any specific area of the project related to this PR?

Uncomment one (or more) /area <> lines:

/area build

/area automation

/area event

/area event_derive

/area plugin

/area plugin_api

/area plugin_derive

/area plugin_tests

What this PR does / why we need it:

I recently changed my github username. I'm just updating it here. I'm also stepping down from approvers and moving myself to emeritus_approvers

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

Does this PR introduce a user-facing change?:

NONE

I recently changed my github username. I'm just updating it here. I'm also stepping down from approvers and moving myself to emeritus_approvers

Signed-off-by: Andrea Terzolo <andreaterzolo3@gmail.com>
@poiana

poiana commented Aug 29, 2026

Copy link
Copy Markdown

There is not a label identifying the kind of this PR.
Please specify it either using /kind <group> or manually from the side menu.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@gnosek

gnosek commented Sep 6, 2026

Copy link
Copy Markdown
Collaborator

/approve :(

@poiana poiana added the approved label Sep 6, 2026
@poiana

poiana commented Sep 6, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: andreaterzolo, gnosek

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana

poiana commented Sep 6, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 3d4e288810f8add27877cb5cb74d3c5340903548

@poiana
poiana merged commit ab61166 into main Sep 6, 2026
7 of 8 checks passed
@poiana
poiana deleted the andreaterzolo-patch-1 branch September 6, 2026 06:41
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants