Skip to content

fix(plugin_api)!: regenerate plugin_api with bindgen 0.73 - #87

Merged
poiana merged 1 commit into
falcosecurity:mainfrom
Jules-Bertholet:bindgen-0.73
Sep 8, 2026
Merged

fix(plugin_api)!: regenerate plugin_api with bindgen 0.73#87
poiana merged 1 commit into
falcosecurity:mainfrom
Jules-Bertholet:bindgen-0.73

Conversation

@Jules-Bertholet

Copy link
Copy Markdown
Contributor

What type of PR is this?

/kind bug

Any specific area of the project related to this PR?

/area plugin_api

What this PR does / why we need it:

Regenerates the FFI bindings in plugin_api with the latest version of rust-bingen. This removes some incorrect derive(Debug)s on newtype wrappers around c_void—these Debug impls were effectively uncallable. See rust-lang/rust#160115 for details.

Does this PR introduce a user-facing change?:

falco_plugin_api : The `Debug` impls for `ss_plugin_table_t`, `ss_plugin_table_entry_t`, `ss_plugin_table_field_t`, `ss_plugin_owner_t`, `ss_plugin_t`, `ss_instance_t`, `ss_plugin_table_iterator_state_t`, `ss_plugin_routine_t`, and `ss_plugin_routine_state_t` have been removed. (These impls were effectively un-callable, as it is never correct to obtain an `&` reference to these types.)

Signed-off-by: Jules Bertholet <julesbertholet@quoi.xyz>
@poiana
poiana requested review from gnosek and mrgian September 6, 2026 05:29
@poiana

poiana commented Sep 6, 2026

Copy link
Copy Markdown

Welcome @Jules-Bertholet! It looks like this is your first PR to falcosecurity/plugin-sdk-rs 🎉

@poiana poiana added the size/S label Sep 6, 2026
@gnosek gnosek changed the title Regenerate plugin_api with bindgen 0.73 fix(plugin_api)!: regenerate plugin_api with bindgen 0.73 Sep 6, 2026
@poiana

poiana commented Sep 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: gnosek, Jules-Bertholet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana poiana added the approved label Sep 8, 2026
@poiana

poiana commented Sep 8, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 976daa6005512f9b38f3162f4856828092b714d6

@poiana
poiana merged commit 598be20 into falcosecurity:main Sep 8, 2026
7 of 10 checks passed
@Jules-Bertholet
Jules-Bertholet deleted the bindgen-0.73 branch September 8, 2026 13:22
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants