Skip to content

chore(deps): bump @t3-oss/env-nextjs from 0.12.0 to 0.13.11 in /apps/merkle-tree-generator - #108

Open
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/apps/merkle-tree-generator/t3-oss/env-nextjs-0.13.11
Open

dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/apps/merkle-tree-generator/t3-oss/env-nextjs-0.13.11

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Apr 9, 2026

Copy link
Copy Markdown
Contributor

Bumps @t3-oss/env-nextjs from 0.12.0 to 0.13.11.

Changelog

Sourced from @​t3-oss/env-nextjs's changelog.

0.13.11

Patch Changes

  • Updated dependencies [6937086]:
    • @​t3-oss/env-core@​0.13.11

0.13.10

Patch Changes

  • Updated dependencies [a778bf3]:
    • @​t3-oss/env-core@​0.13.10

0.13.9

Patch Changes

  • Updated dependencies [5987d5a]:
    • @​t3-oss/env-core@​0.13.9

0.13.8

Patch Changes

0.13.7

Patch Changes

0.13.6

Patch Changes

  • Updated dependencies [aa245ed]:
    • @​t3-oss/env-core@​0.13.6

... (truncated)

Commits
Maintainer changes

This version was pushed to npm by [GitHub Actions](https://www.npmjs.com/~GitHub Actions), a new releaser for @​t3-oss/env-nextjs since your current version.


Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [@t3-oss/env-nextjs](https://github.com/t3-oss/t3-env/tree/HEAD/packages/nextjs) from 0.12.0 to 0.13.11.
- [Release notes](https://github.com/t3-oss/t3-env/releases)
- [Changelog](https://github.com/t3-oss/t3-env/blob/main/packages/nextjs/CHANGELOG.md)
- [Commits](https://github.com/t3-oss/t3-env/commits/@t3-oss/env-nextjs@0.13.11/packages/nextjs)

---
updated-dependencies:
- dependency-name: "@t3-oss/env-nextjs"
  dependency-version: 0.13.11
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot @github

dependabot Bot commented on behalf of github Apr 9, 2026

Copy link
Copy Markdown
Contributor Author

Labels

The following labels could not be found: app:merkle-tree-generator, dependencies. Please create them before Dependabot can add them to a pull request.

Please fix the above issues or remove invalid values from dependabot.yml.

@vercel

vercel Bot commented Apr 9, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

4 Skipped Deployments
Project Deployment Actions Updated (UTC)
merkle-tree-generator Ignored Ignored Preview Apr 9, 2026 10:09pm
donation Skipped Skipped Apr 9, 2026 10:09pm
landing-page Skipped Skipped Apr 9, 2026 10:09pm
onchain-payment Skipped Skipped Apr 9, 2026 10:09pm

Request Review

@vercel
vercel Bot temporarily deployed to Preview – landing-page April 9, 2026 22:09 Inactive
@vercel
vercel Bot temporarily deployed to Preview – onchain-payment April 9, 2026 22:09 Inactive
@vercel
vercel Bot temporarily deployed to Preview – donation April 9, 2026 22:09 Inactive

This branch was previously deployed

3 inactive deployments
Preview – landing-page cb6f2627 Deployed Apr 9, 2026 by vercel[bot]
Preview – onchain-payment cb6f2627 Deployed Apr 9, 2026 by vercel[bot]
Preview – donation cb6f2627 Deployed Apr 9, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants