Skip to content

Use separate maps for selector/regular labels - #29

Open
akalenyu wants to merge 1 commit into
kubevirt:masterfrom
akalenyu:distinct-selector
Open

Use separate maps for selector/regular labels#29
akalenyu wants to merge 1 commit into
kubevirt:masterfrom
akalenyu:distinct-selector

Conversation

@akalenyu

@akalenyu akalenyu commented Aug 4, 2026

Copy link
Copy Markdown
Contributor

What this PR does / why we need it:
This bit us at kubevirt/kubevirt-migration-operator#85

Which issue(s) this PR fixes (optional, in fixes #<issue number>(, fixes #<issue_number>, ...) format, will close the issue(s) when PR gets merged):
Fixes #

Special notes for your reviewer:

Checklist

This checklist is not enforcing, but it's a reminder of items that could be relevant to every PR.
Approvers are expected to review this list.

Release note:

NONE

This bit us at kubevirt/kubevirt-migration-operator#85

Signed-off-by: Alex Kalenyuk <akalenyu@redhat.com>
@kubevirt-prow kubevirt-prow Bot added release-note-none Denotes a PR that doesn't merit a release note. dco-signoff: yes Indicates the PR's author has DCO signed all their commits. labels Aug 4, 2026
@kubevirt-prow
kubevirt-prow Bot requested review from awels and mhenriks August 4, 2026 10:17
@kubevirt-prow kubevirt-prow Bot added the size/M label Aug 4, 2026
@akalenyu

akalenyu commented Aug 4, 2026

Copy link
Copy Markdown
Contributor Author

/approve

@kubevirt-prow kubevirt-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 4, 2026
@akalenyu

akalenyu commented Aug 4, 2026

Copy link
Copy Markdown
Contributor Author

/approve cancel

@kubevirt-prow

kubevirt-prow Bot commented Aug 4, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign mhenriks for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubevirt-prow kubevirt-prow Bot removed the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 4, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes Indicates the PR's author has DCO signed all their commits. release-note-none Denotes a PR that doesn't merit a release note. size/M

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant