chore(deps): update module github.com/evanphx/json-patch to v5 - #294
Conversation
|
Hi @red-hat-konflux-kflux-prd-rh03[bot]. Thanks for your PR. I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## master #294 +/- ##
=======================================
Coverage 65.06% 65.06%
=======================================
Files 5 5
Lines 541 541
=======================================
Hits 352 352
Misses 152 152
Partials 37 37 🚀 New features to boost your workflow:
|
15249de to
e65cf50
Compare
e65cf50 to
5f863ea
Compare
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: red-hat-konflux-kflux-prd-rh03[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
1 similar comment
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: red-hat-konflux-kflux-prd-rh03[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
5f863ea to
410b53e
Compare
6e1ccb5 to
8d98879
Compare
8d98879 to
0e12354
Compare
0e12354 to
0b4aec1
Compare
0b4aec1 to
ec48c46
Compare
ec48c46 to
a7c0eef
Compare
97a1e49 to
a7c0eef
Compare
Signed-off-by: red-hat-konflux-kflux-prd-rh03 <206760901+red-hat-konflux-kflux-prd-rh03[bot]@users.noreply.github.com>
This PR contains the following updates:
v4.12.0+incompatible→v5.9.11+incompatibleWarning
Some dependencies could not be looked up. Check the Dependency Dashboard for more information.
Release Notes
evanphx/json-patch (github.com/evanphx/json-patch)
v5.9.11+incompatibleCompare Source
v5.9.10+incompatibleCompare Source
v5.9.0+incompatibleCompare Source
v5.8.1+incompatibleCompare Source
v5.8.0+incompatibleCompare Source
v5.7.0+incompatibleCompare Source
v5.6.0+incompatibleCompare Source
v4.13.0+incompatibleCompare Source
Configuration
📅 Schedule: (in timezone UTC)
* 2-4 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
To execute skipped test pipelines write comment
/ok-to-test.Documentation
Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.